Skip to content

chore: add CODEOWNERS with maintainers team - #60

Merged
setrofim merged 1 commit into
mainfrom
org-teams-codeowners
Oct 5, 2026
Merged

setrofim merged 1 commit into
mainfrom
org-teams-codeowners

Conversation

@THS-on

@THS-on THS-on commented Oct 3, 2026

Copy link
Copy Markdown
Member

No description provided.

Signed-off-by: Thore Sommer <mail@thson.de>
@setrofim
setrofim merged commit 9771e2b into main Oct 5, 2026
7 checks passed
@setrofim
setrofim deleted the org-teams-codeowners branch October 5, 2026 08:26
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants