Skip to content

chore: update CODEOWNERS - #287

Merged
FrostByte0x merged 1 commit into
mainfrom
rbeuque74-patch-1
Sep 22, 2026
Merged

FrostByte0x merged 1 commit into
mainfrom
rbeuque74-patch-1

Conversation

@rbeuque74

Copy link
Copy Markdown
Member

No description provided.

Signed-off-by: Romain Beuque <556072+rbeuque74@users.noreply.github.com>
@rbeuque74
rbeuque74 requested a review from a team as a code owner September 22, 2026 10:13
@FrostByte0x
FrostByte0x merged commit e79d06c into main Sep 22, 2026
7 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants